Skip to content

Add default CODEOWNERS for repository (#63) - #64

Closed
robertcolbert wants to merge 1 commit into
developfrom
main
Closed

robertcolbert wants to merge 1 commit into
developfrom
main

Conversation

@robertcolbert

Copy link
Copy Markdown
Collaborator

This minor change need to requires develop be rebased on main.

@google-cla

google-cla Bot commented Sep 9, 2026

Copy link
Copy Markdown

Thanks for your pull request! It looks like this may be your first contribution to a Google open source project. Before we can look at your pull request, you'll need to sign a Contributor License Agreement (CLA).

View this failed invocation of the CLA check for more information.

For the most up to date status, view the checks section at the bottom of the pull request.

@robertcolbert

Copy link
Copy Markdown
Collaborator Author

We decide not to use develop

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

2 participants